Skip to content

fosrl-pangolin: 1.2.0 -> 1.8.0 - #419533

Merged
NickCao merged 1 commit into
NixOS:masterfrom
SigmaSquadron:push-sqwpmnnxpymt
Aug 6, 2025
Merged

fosrl-pangolin: 1.2.0 -> 1.8.0#419533
NickCao merged 1 commit into
NixOS:masterfrom
SigmaSquadron:push-sqwpmnnxpymt

Conversation

@SigmaSquadron

@SigmaSquadron SigmaSquadron commented Jun 24, 2025

Copy link
Copy Markdown
Contributor

Changelogs: 1.3.0, 1.3.1, 1.3.2, 1.4.0, 1.5.0, 1.5.1, 1.6.0, 1.6.1, 1.6.2, 1.7.0, 1.7.1, 1.7.2, 1.7.3, 1.8.0.

This combined update introduces a new database type for Pangolin: PostgreSQL. The module for it will come at a later date, so no separate package for it right now.

1.6.0 also introduces the pangctl command, which has inspired me to write the spaghetti you see in the preFixup section of the diff. By wrapping the *.mjs files to actual commands in $out/bin, we can simplify the NixOS module, make it easier to access pangctl, and allow non-NixOS users to start a Pangolin server without much hassle.

Things done

  • Built on platform(s)
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • For non-Linux: Is sandboxing enabled in nix.conf? (See Nix manual)
    • sandbox = relaxed
    • sandbox = true
  • Tested, as applicable:
  • Tested compilation of all packages that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage
  • Tested basic functionality of all binary files (usually in ./result/bin/)
  • Nixpkgs 25.11 Release Notes (or backporting 24.11 and 25.05 Nixpkgs Release notes)
    • (Package updates) Added a release notes entry if the change is major or breaking
  • NixOS 25.11 Release Notes (or backporting 24.11 and 25.05 NixOS Release notes)
    • (Module updates) Added a release notes entry if the change is significant
    • (Module addition) Added a release notes entry if adding a new NixOS module
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other contributing documentation in corresponding paths.

Add a 👍 reaction to pull requests you find important.

@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. labels Jun 24, 2025

@jackrosenberg jackrosenberg left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Lgtm

@jackrosenberg jackrosenberg added the 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. label Jun 24, 2025
@nixpkgs-ci nixpkgs-ci Bot added the 12.approvals: 1 This PR was reviewed and approved by one person. label Jun 24, 2025
@SigmaSquadron
SigmaSquadron force-pushed the push-sqwpmnnxpymt branch 6 times, most recently from 9093f73 to 267360c Compare July 6, 2025 11:09
@SigmaSquadron SigmaSquadron changed the title fosrl-pangolin: 1.2.0 -> 1.5.1; fosrl-pangolin-postgres: init at 1.5.1 fosrl-pangolin: 1.2.0 -> 1.6.2; fosrl-pangolin-postgres: init at 1.6.2 Jul 6, 2025
@SigmaSquadron

SigmaSquadron commented Jul 6, 2025

Copy link
Copy Markdown
Contributor Author

Updated to 1.6.2. The highlight of this update is the new commands in $out/bin. I've also taken the opportunity to move the Pangolin scripts and static files to $out/share/pangolin so we can have some semblance of FHS paths instead of the mess NextJS forces unto us.

Comment thread pkgs/by-name/fo/fosrl-pangolin/package.nix Outdated
Comment thread pkgs/by-name/fo/fosrl-pangolin/package.nix Outdated
@SigmaSquadron
SigmaSquadron force-pushed the push-sqwpmnnxpymt branch 2 times, most recently from 597a9b1 to 1e64d3c Compare July 8, 2025 06:56
@jackrosenberg jackrosenberg mentioned this pull request Jul 17, 2025
13 tasks
@jackrosenberg
jackrosenberg self-requested a review July 20, 2025 10:08

@jackrosenberg jackrosenberg left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Revoked approval since nixos/pangolin needs to be updated at the same time as updating to 1.6 and above.

@jackrosenberg jackrosenberg removed 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. labels Jul 20, 2025
@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. labels Jul 20, 2025
@SigmaSquadron SigmaSquadron changed the title fosrl-pangolin: 1.2.0 -> 1.6.2; fosrl-pangolin-postgres: init at 1.6.2 fosrl-pangolin: 1.2.0 -> 1.7.3; fosrl-pangolin-postgres: init at 1.7.3 Jul 21, 2025
@jackrosenberg
jackrosenberg self-requested a review July 23, 2025 14:36
@SigmaSquadron SigmaSquadron changed the title fosrl-pangolin: 1.2.0 -> 1.7.3; fosrl-pangolin-postgres: init at 1.7.3 fosrl-pangolin: 1.2.0 -> 1.7.3 Jul 23, 2025
@nixpkgs-ci nixpkgs-ci Bot added the 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. label Jul 23, 2025
@jackrosenberg

Copy link
Copy Markdown
Member

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 419533
Commit: d299df66b63ba4e6f2f98f5b1e13bedd7b748d8b


x86_64-linux

✅ 1 package built:
  • fosrl-pangolin

1 similar comment
@ethancedwards8

Copy link
Copy Markdown
Member

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 419533
Commit: d299df66b63ba4e6f2f98f5b1e13bedd7b748d8b


x86_64-linux

✅ 1 package built:
  • fosrl-pangolin

Comment thread pkgs/by-name/fo/fosrl-pangolin/package.nix Outdated
Comment thread pkgs/by-name/fo/fosrl-pangolin/package.nix Outdated
@SigmaSquadron
SigmaSquadron force-pushed the push-sqwpmnnxpymt branch 2 times, most recently from 532ade7 to 51eeaf5 Compare July 27, 2025 05:54
@SigmaSquadron
SigmaSquadron requested a review from NickCao July 27, 2025 06:00
- This update introduces the PostgreSQL database type, which can be
  selected instead of SQLite.
- The derivation is now using `finalAttrs`, as buildNpmPackage supports
  it, so there's no reason to use `rec`.
- Finally, in order to simplify the module and allow non-NixOS users to
  use the Pangolin server, the minified `.mjs` files have been wrapped
  into shell scripts that can set up an appropriate environment for the
  server.

Signed-off-by: Fernando Rodrigues <alpha@sigmasquadron.net>
@SigmaSquadron SigmaSquadron changed the title fosrl-pangolin: 1.2.0 -> 1.7.3 fosrl-pangolin: 1.2.0 -> 1.8.0 Aug 1, 2025
@jackrosenberg

Copy link
Copy Markdown
Member

@NickCao would you mind taking another look at this?

@NickCao

NickCao commented Aug 1, 2025

Copy link
Copy Markdown
Member

Still not very convinced by the make it work out of nixos part (since most of these users would happily use docker instead), but otherwise lgtm.

@jackrosenberg

Copy link
Copy Markdown
Member

Still not very convinced by the make it work out of nixos part (since most of these users would happily use docker instead), but otherwise lgtm.

Yea, it's more for the people who use Nix + another distro already. We'll rewrite this as one of the new modular services at some point in the future anyway.

@SigmaSquadron

Copy link
Copy Markdown
Contributor Author

Well, this also has the benefit of making the module simpler. The ollama-nextjs-webui package also has something similar.

@jackrosenberg

Copy link
Copy Markdown
Member

@adamcstephens, would you mind taking a look at this? It's part of the pangolin stack, for which you reviewed olm

@NickCao

NickCao commented Aug 6, 2025

Copy link
Copy Markdown
Member

Sorry this got lost in my pile of backlog.

@NickCao
NickCao merged commit 7c211ac into NixOS:master Aug 6, 2025
27 of 29 checks passed
@jackrosenberg

Copy link
Copy Markdown
Member

No worries, thanks! This unblocks #416148, which i will rebase in an hour or so :)

@SigmaSquadron
SigmaSquadron deleted the push-sqwpmnnxpymt branch August 13, 2025 08:59
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants