Skip to content

fix(e2e) Loki configuration in e2e test - #1480

Open
enarha wants to merge 1 commit into
tektoncd:mainfrom
enarha:fix-e2e-loki-config
Open

fix(e2e) Loki configuration in e2e test#1480
enarha wants to merge 1 commit into
tektoncd:mainfrom
enarha:fix-e2e-loki-config

Conversation

@enarha

@enarha enarha commented Sep 6, 2026

Copy link
Copy Markdown
Contributor

It was broken in multiple ways, typo in the env var name, deprecated playground option, duplicated path, incompatible limit and pod label mapping. This only affects the e2e flow with loki log forwarding enabled and does not affect the general e2e tests as used in ci.

Changes

/kind misc

Submitter Checklist

These are the criteria that every PR should meet, please check them off as you review them:

  • Has Docs included if any changes are user facing
  • Has Tests included if any functionality added or changed
  • Tested your changes locally (if this is a code change)
  • Follows the commit message standard
  • Meets the Tekton contributor standards (including functionality, content, code)
  • Has a kind label. You can add a comment on this PR that contains /kind <type>. Valid types are bug, cleanup, design, documentation, feature, flake, misc, question, tep
  • Release notes block below has been updated with any user-facing changes (API changes, bug fixes, changes requiring upgrade notices or deprecation warnings)
  • Release notes contain the string "action required" if the change requires additional action from users switching to the new release

Release Notes

NONE

It was broken in multiple ways, typo in the env var name, deprecated
`playground` option, duplicated path, incompatible limit and pod label
mapping. This only affects the e2e flow with loki log forwarding enabled
and does not affect the general e2e tests as used in ci.
@tekton-robot tekton-robot added release-note Denotes a PR that will be considered when it comes time to generate release notes. kind/misc Categorizes issue or PR as a miscellaneuous one. labels Sep 6, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
To complete the pull request process, please ask for approval from enarha after the PR has been reviewed.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@enarha
enarha removed the request for review from divyansh42 September 6, 2026 12:07
@tekton-robot tekton-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 6, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/misc Categorizes issue or PR as a miscellaneuous one. release-note Denotes a PR that will be considered when it comes time to generate release notes. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants