Skip to content

fix: remove npm update in ci - #4001

Merged
dnechay merged 2 commits into
developfrom
dnechay/npm-ci
Aug 10, 2026
Merged

dnechay merged 2 commits into
developfrom
dnechay/npm-ci

Conversation

@dnechay

@dnechay dnechay commented Aug 10, 2026 •

Copy link
Copy Markdown
Member

Issue tracking

N/A

Context behind the change

"Publish packages" action fails because of npm update step in it (ref). Some time ago when we were using Node@22 - default npm version was below 11.5.1 required for OIDC. Now we use 24.13 and it already includes npm@11.6.2, so we can just remove unnecessary step.

Also removed unnecessary env var for HUMAN App

How has this been tested?

  • Manually triggered action from this branch (ref)

Release plan

Merge to main

Potential risks; What to monitor; Rollback plan

No

@vercel

vercel Bot commented Aug 10, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated (UTC)
human-app Ready Ready Preview Aug 10, 2026 12:49pm
4 Skipped Deployments
Project Deployment Actions Updated (UTC)
faucet-frontend Ignored Ignored Preview Aug 10, 2026 12:49pm
faucet-server Ignored Ignored Preview Aug 10, 2026 12:49pm
human-dashboard-frontend Skipped Skipped Aug 10, 2026 12:49pm
staking-dashboard Skipped Skipped Aug 10, 2026 12:49pm

Request Review

@dnechay dnechay self-assigned this Aug 10, 2026
@dnechay
dnechay requested a review from Dzeranov August 10, 2026 11:55
Dzeranov
Dzeranov previously approved these changes Aug 10, 2026
@vercel
vercel Bot temporarily deployed to Preview – human-dashboard-frontend August 10, 2026 12:47 Inactive
@vercel
vercel Bot temporarily deployed to Preview – staking-dashboard August 10, 2026 12:47 Inactive
@dnechay
dnechay requested a review from KirillKirill August 10, 2026 12:48
@dnechay
dnechay merged commit d7eef08 into develop Aug 10, 2026
13 checks passed
@dnechay
dnechay deleted the dnechay/npm-ci branch August 10, 2026 12:51
@dnechay dnechay mentioned this pull request Aug 10, 2026
1 task done

This branch was successfully deployed

1 active and 2 inactive deployments
Preview – human-app — 144d5b41 Deployed Aug 10, 2026 by vercel[bot]
Preview – human-dashboard-frontend — 144d5b41 Deployed Aug 10, 2026 by vercel[bot]
Preview – staking-dashboard — 144d5b41 Deployed Aug 10, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants