Skip to content

nixos/traefik: add plugins option to install Nix-managed plugins - #436793

Merged
SigmaSquadron merged 1 commit into
NixOS:wip-traefik-pluginsfrom
SigmaSquadron:push-pnzvtsykmttt
Aug 26, 2025
Merged

SigmaSquadron merged 1 commit into
NixOS:wip-traefik-pluginsfrom
SigmaSquadron:push-pnzvtsykmttt

Conversation

@SigmaSquadron

@SigmaSquadron SigmaSquadron commented Aug 25, 2025

Copy link
Copy Markdown
Contributor

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

Add a 👍 reaction to pull requests you find important.

Comment thread nixos/modules/services/web-servers/traefik.nix Outdated
Comment thread nixos/modules/services/web-servers/traefik.nix Outdated

@jackrosenberg jackrosenberg left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Will need to test this later, but looks good!

@nixpkgs-ci nixpkgs-ci Bot added the 12.approvals: 1 This PR was reviewed and approved by one person. label Aug 25, 2025
@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 6.topic: nixos Issues or PRs affecting NixOS modules, or package usability issues specific to NixOS 8.has: module (update) This PR changes an existing module in `nixos/` labels Aug 25, 2025
Comment thread nixos/modules/services/web-servers/traefik.nix Outdated
Signed-off-by: Fernando Rodrigues <alpha@sigmasquadron.net>
@SigmaSquadron
SigmaSquadron merged commit cb00274 into NixOS:wip-traefik-plugins Aug 26, 2025
29 of 31 checks passed
@SigmaSquadron
SigmaSquadron deleted the push-pnzvtsykmttt branch August 26, 2025 09:58
@adamcstephens

Copy link
Copy Markdown
Contributor

I'd probably suggest waiting a few days, and giving people time to review, before merging something like this.

@jackrosenberg

Copy link
Copy Markdown
Member

I'd probably suggest waiting a few days, and giving people time to review, before merging something like this.

@adamcstephens This isn't master. I asked him to merge both this and other PR's from #436807 :)

@adamcstephens

Copy link
Copy Markdown
Contributor

I see. Well, it's still not even working, so it's still a bit odd to merge. Traefik is picky about its config, and this breaks configs without plugins.

From nix build -f . traefik.tests

vm-test-run-traefik> traefik # [   17.158990] traefik[1178]: {"level":"error","error":"command traefik error: localPlugins cannot be a standalone element (type map[string]pl

@SigmaSquadron

SigmaSquadron commented Aug 26, 2025

Copy link
Copy Markdown
Contributor Author

Yes, that was the goal of merging it alongside the other PRs. I would never actually merge something as work-in-progress as this into master, which is why the branch is prefixed with wip-. We've also found other issues that can be fixed iteratively.

@SigmaSquadron

Copy link
Copy Markdown
Contributor Author

This is slightly more ergonomic than me giving @jackrosenberg commit rights to my Nixpkgs fork and us working outside of the NixOS repo.

@adamcstephens

Copy link
Copy Markdown
Contributor

Yep. Sorry I missed it wasn’t master. :)

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

6.topic: nixos Issues or PRs affecting NixOS modules, or package usability issues specific to NixOS 8.has: module (update) This PR changes an existing module in `nixos/` 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 12.approvals: 1 This PR was reviewed and approved by one person.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants