Skip to content

freecad: 1.0.1 -> 1.0.2 - #431206

Merged
pbsds merged 3 commits into
NixOS:masterfrom
LordGrimmauld:freecad_1_0_2
Aug 20, 2025
Merged

freecad: 1.0.1 -> 1.0.2#431206
pbsds merged 3 commits into
NixOS:masterfrom
LordGrimmauld:freecad_1_0_2

Conversation

@LordGrimmauld

@LordGrimmauld LordGrimmauld commented Aug 5, 2025

Copy link
Copy Markdown
Contributor

Tag: https://github.com/FreeCAD/FreeCAD/releases/tag/1.0.2

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

Add a 👍 reaction to pull requests you find important.

@nixpkgs-ci nixpkgs-ci Bot added 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. labels Aug 5, 2025
@LordGrimmauld LordGrimmauld reopened this Aug 5, 2025
@nixpkgs-ci nixpkgs-ci Bot added the 9.needs: reviewer This PR currently has no reviewers requested and needs attention. label Aug 5, 2025
@grimmauld-bot

Copy link
Copy Markdown

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 431206
Commit: 7f9a94dea98f24a2b6784d97da3bfc9a9faeac69


x86_64-linux

❌ 1 package failed to build:
  • freecad

@LordGrimmauld

Copy link
Copy Markdown
Contributor Author

huh, they force-pushed the tag... Pain.

@grimmauld-bot

Copy link
Copy Markdown

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 431206
Commit: 6d8e32ef2d800ca1b40b4e14fc8a64974f25a6da


x86_64-linux

✅ 1 package built:
  • freecad

@qbisi

qbisi commented Aug 11, 2025

Copy link
Copy Markdown
Contributor
  1. In fact we can safely remove mpi in BuildInputs no matter how we set FREECAD_USE_SMESH.
  2. Setting FREECAD_USE_SMESH = OFF does not work, switch to BUILD_FEM = OFF instead. see https://github.com/FreeCAD/FreeCAD/blob/79738eec3812b2a85cdf7a8a145dcd8f3f28eaaf/cMake/FreeCAD_Helpers/InitializeFreeCADBuildOptions.cmake#L191.
  3. Build_FEM require hdf5, vtk, medfile

The whole salomePackages sucks.

@LordGrimmauld

Copy link
Copy Markdown
Contributor Author

huh. Thats... interesting, and kinda disappointing.

Comment thread pkgs/by-name/fr/freecad/package.nix Outdated
@nixpkgs-ci nixpkgs-ci Bot removed the 9.needs: reviewer This PR currently has no reviewers requested and needs attention. label Aug 11, 2025
@LordGrimmauld

Copy link
Copy Markdown
Contributor Author

Okay, updated this PR. We can't do netgen yet, that does still segfault. However, we can add gmsh instead, that seems to be in a decent state of maintenance in nixpkgs and should at least alleviate some of the requirements: We now do have a mesher. smesh is still there, but mpi can be removed aiui.

Comment thread pkgs/by-name/fr/freecad/package.nix Outdated

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

how about adding gmsh in binPath, can avoid patch?

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

i tried, not quite - it will make the "get gmsh version" popup dialog fail with a nasty error message, despite gmsh actually being present. Because that checks whether it has an absolute path.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

image

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

(gmsh is also probably not reproducible considering the info it collects here. But that is not my concern currently.)

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

cmake respect SOURCE_DATE_EPOCH which will be set by set-source-date-epoch-to-latest.sh in stdenv.
2025-07-03 is right the time when gmsh-1.14.0 was released.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I install gmsh via nix profile install nixpkgs#gmsh in my path and i can use the gmsh plugin in freecad.
IMO we don't have to bundle gmsh and we can leave the choice to users.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

i disagree. We should supply the plugin things and make them explicitly available. Otherwise these aren't at all discoverable. The compromise would be adding a freecad wrapper package that supplies these plugins without a full rebuild, that i could agree with. With the smesh thing segfaulting (#408577), we should imo provide an easy alternative.

According to upstream, "Med Fichier can require MPI".
We do not have Med Fichier, so we can remove mpi.
If this causes issues, it should be reverted. However,
FEM still works, so this is probably fine.
@LordGrimmauld

Copy link
Copy Markdown
Contributor Author

Rebased to fix merge conflicts with #434105

@nixos-discourse

Copy link
Copy Markdown

This pull request has been mentioned on NixOS Discourse. There might be relevant details there:

https://discourse.nixos.org/t/which-package-for-freecad/67961/9

@pbsds
pbsds merged commit f6335dd into NixOS:master Aug 20, 2025
26 of 28 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants